Conversation
|
Skipping CI for Draft Pull Request. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Draft PR not reviewedDraft PRs are not automatically reviewed by default.
To automatically review draft PRs, update your CodeRabbit configuration: reviews:
auto_review:
drafts: trueComment |
|
Zuul encountered a syntax error while parsing its The nodeset "centos-10-medium-2x-centos-10-crc-cloud-ocp-4-20-1-3xl" The problem appears in the "watcher-operator-base-centos-10" job stanza: job: in "openstack-k8s-operators/watcher-operator/.zuul.yaml@main", line 22 |
|
Zuul encountered a syntax error while parsing its The nodeset "centos-10-medium-2x-centos-10-crc-cloud-ocp-4-20-1-3xl" The problem appears in the "watcher-operator-base-centos-10" job stanza: job: in "openstack-k8s-operators/watcher-operator/.zuul.yaml@main", line 22 |
|
This change depends on a change that failed to merge. Change openstack-k8s-operators/ci-framework#4163 is needed. |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 45m 15s |
|
recheck |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 22m 28s |
|
most of the live migration tests are failing https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/logs//e8a/rdoproject.org/e8ae786d8cef40af8e26e93a8dcebece/controller/ci-framework-data/tests/test_operator/tempest-tests-tempest/tempest_results.html.gz One of the example of zone migration test Based on compute-0 log in sosreport-compute-0/sosreport-compute-0-2026-09-15-vliyjop/var/log/messages It might be network connectvity issue between two nodes. |
|
recheck after node hold |
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 10m 03s |
More logs: |
|
|
Build failed (check pipeline). Post ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 31m 11s |
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 20m 45s |
|
Zuul encountered a syntax error while parsing its The nodeset "centos-10-medium-2x-centos-10-crc-cloud-ocp-4-20-1-3xl" The problem appears in the "watcher-operator-base-centos-10" job stanza: job: in "openstack-k8s-operators/watcher-operator/.zuul.yaml@main", line 22 |
Depends-On: openstack-k8s-operators/ci-framework#4163 Signed-off-by: Chandan <raukadah@gmail.com>
|
Build succeeded (check pipeline). ✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 24m 39s |
Depends-On: openstack-k8s-operators/ci-framework#4163