feat: allow setting podLabels on hyperdx Pods - #279
Open
poblahblahblah wants to merge 1 commit into
Open
poblahblahblah wants to merge 1 commit into
poblahblahblah wants to merge 1 commit into
Conversation
🦋 Changeset detectedLatest commit: 924471b The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
jordan-simonovski
requested changes
Sep 16, 2026
jordan-simonovski
left a comment
There was a problem hiding this comment.
Changes are looking good! We just need a changeset for this yarn changeset
poblahblahblah
force-pushed
the
allow-for-podLabels
branch
from
September 16, 2026 22:26
36ff37e to
ccacf72
Compare
This change allows a consumer to add additional podLabels to the hyperdx Pods. Signed-off-by: Patrick O’Brien <patrick.obrien@thetradedesk.com>
poblahblahblah
force-pushed
the
allow-for-podLabels
branch
from
September 16, 2026 22:29
ccacf72 to
924471b
Compare
teeohhem
approved these changes
Sep 18, 2026
teeohhem
enabled auto-merge (squash)
September 18, 2026 14:33
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This change allows a consumer to add additional podLabels to the hyperdx Pods.